Skip to content

Set thread_context_sharing feature and owner to asm - #7916

Draft
danielmercier wants to merge 1 commit into
mainfrom
daniel.mercier/thread-context-sharing-feature
Draft

danielmercier wants to merge 1 commit into
mainfrom
daniel.mercier/thread-context-sharing-feature

Conversation

@danielmercier

Copy link
Copy Markdown
Contributor

Motivation

Properly track the test

Changes

Assign the feature parity # to the thread_context_sharing test

@github-actions

github-actions Bot commented Oct 1, 2026

Copy link
Copy Markdown
Contributor

CODEOWNERS have been resolved as:

tests/cws/test_thread_context_sharing.py                                @DataDog/system-tests-reviewers
utils/_features.py                                                      @DataDog/system-tests-core

@datadog-prod-us1-3

datadog-prod-us1-3 Bot commented Oct 1, 2026 •

Copy link
Copy Markdown
Contributor

Pipelines  Tests

⚠️ Warnings

Your PR has warnings. Please review the issues below.

🚦 5 Pipeline jobs failed

Testing the test | System Tests (golang, dev) / End-to-end #1 / haproxy 1

View more details · View in GitHub Actions

Testing the test | System Tests (nodejs, dev) / End-to-end #1 / express5 1

View more details · View in GitHub Actions

Testing the test | System Tests (php, dev) / Build end-to-end (laravel11x)

View more details · View in GitHub Actions

View all 5 failed jobs.

ℹ️ Info

No other issues found (see more)

🧪 All tests passed
❄️ No new flaky tests detected

Useful? React with 👍 / 👎

This comment will be updated automatically if new data arrives.
🔗 Commit SHA: ef5b84f | Docs | View more details | Give us feedback!

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant