Skip to content

test(otel): Enable OTel semantics tests for dd-trace-dotnet - #7918

Draft
zacharycmontoya wants to merge 1 commit into
mainfrom
zach.montoya/otel-enable-http
Draft

zacharycmontoya wants to merge 1 commit into
mainfrom
zach.montoya/otel-enable-http

Conversation

@zacharycmontoya

Copy link
Copy Markdown
Contributor

Motivation

We've added support for OTel semantics in dd-trace-dotnet v3.53.0 so enabling the test ensures we are exercising the correct behavior and we do not regress it.

Changes

Updates manifests/dotnet.yml to run the relevant test cases from tests/test_otel_http_semantics.py against v3.53.0.

Workflow

  1. ⚠️ Create your PR as draft ⚠️
  2. Work on you PR until the CI passes
  3. Mark it as ready for review
    • Tests, manifest, weblog are modified -> you'll need a review from system-tests-reviewers: ask to one of youre co-worker familiar with the tested feature.
    • Framework is modified, or non obvious usage of it -> get a review from system-tests-core (slack)

🚀 Once your PR is reviewed and the CI green, you can merge it!

🛟 #apm-shared-testing 🛟

@github-actions

github-actions Bot commented Oct 1, 2026

Copy link
Copy Markdown
Contributor

CODEOWNERS have been resolved as:

manifests/dotnet.yml                                                    @DataDog/system-tests-reviewers

@datadog-official

datadog-official Bot commented Oct 1, 2026 •

Copy link
Copy Markdown

Pipelines  Tests

❌ Errors

Your PR has failed checks. Please review the issues below and take necessary action before merging.

🚦 5 Pipeline jobs failed

Testing the test | System Tests (dotnet, dev) / End-to-end #1 / poc 1

View more details · View in GitHub Actions

Testing the test | System Tests (dotnet, dev) / End-to-end #1 / uds 1

View more details · View in GitHub Actions

Testing the test | System Tests (dotnet, prod) / End-to-end #1 / poc 1

View more details · View in GitHub Actions

View all 5 failed jobs.

ℹ️ Info

No other issues found (see more)

🧪 All tests passed
❄️ No new flaky tests detected

Useful? React with 👍 / 👎

This comment will be updated automatically if new data arrives.
🔗 Commit SHA: 7a7795e | Docs | View more details | Give us feedback!

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant