Skip to content

CODEOWNERS file added - #1

Open
nxvl wants to merge 1 commit into
mainfrom
codeowners
Open

nxvl wants to merge 1 commit into
mainfrom
codeowners

Conversation

@nxvl

@nxvl nxvl commented Nov 4, 2022

Copy link
Copy Markdown

SUMMARY
CODEOWNERS file added to the repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant