fix(deps): update dependency web-vitals to v4 - #477
descope[bot] wants to merge 1 commit into
Conversation
|
The latest updates on your projects. Learn more about Vercel for Git ↗︎
5 Skipped Deployments
|
9af49fb to
95956f9
Compare
0a78d01 to
6b9e518
Compare
6b9e518 to
fcc9553
Compare
fcc9553 to
27b861b
Compare
27b861b to
c08cfe8
Compare
dfd0561 to
1d71f5b
Compare
Edited/Blocked NotificationRenovate will not automatically rebase this PR, because it does not recognize the last commit author and assumes somebody else may have edited the PR. You can manually request rebase by checking the rebase/retry box above. |
Autoclosing SkippedThis PR has been flagged for autoclosing. However, it is being skipped due to the branch being already modified. Please close/delete it manually or report a bug if you think this is in error. |
1d71f5b to
4957fa0
Compare
|
Your Vercel team allenzhou is not permitted to deploy from this git repository. Contact an administrator to add github organization descope as a Protected Git Scope in allenzhou on Vercel. Once added, commit again to see your changes. Learn more: https://vercel.com/docs/security/protected-git-scopes |
4957fa0 to
f33f381
Compare
This PR contains the following updates:
3.5.2→4.2.4Warning
Some dependencies could not be looked up. Check the Dependency Dashboard for more information.
Release Notes
GoogleChrome/web-vitals (web-vitals)
v4.2.4Compare Source
v4.2.3Compare Source
v4.2.2Compare Source
v4.2.1Compare Source
v4.2.0Compare Source
v4.1.1Compare Source
v4.1.0Compare Source
v4.0.1Compare Source
ReportCallbacktype back but deprecate it (#483)v4.0.0Compare Source
waitingDurationto make it easier to understand redirect delays (#458)TTFBAttributionfields from*Timeto*Duration(#453)resourceLoadTimetoresourceLoadDurationin LCP attribution (#450)onFID()and remove previously deprecated APIs (#435)reportAllChangesbehavior for LCP when library is loaded late (#468)Configuration
📅 Schedule: (in timezone Asia/Jerusalem)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Never, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR has been generated by Mend Renovate.