fix(deps): update material-ui monorepo to v7 - #811
Open
descope[bot] wants to merge 1 commit into
Open
descope[bot] wants to merge 1 commit into
descope[bot] wants to merge 1 commit into
Conversation
|
The latest updates on your projects. Learn more about Vercel for Git ↗︎
6 Skipped Deployments
|
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
April 16, 2025 22:42
e3be09a to
1048cca
Compare
|
Your Vercel team allenzhou is not permitted to deploy from this git repository. Contact an administrator to add github organization descope as a Protected Git Scope in allenzhou on Vercel. Once added, commit again to see your changes. Learn more: https://vercel.com/docs/security/protected-git-scopes |
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
April 17, 2025 02:03
1048cca to
e921dd1
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
April 18, 2025 19:09
e921dd1 to
d1d2270
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
April 18, 2025 20:49
d1d2270 to
07c58dc
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
April 19, 2025 21:02
07c58dc to
bb88407
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
April 20, 2025 13:50
bb88407 to
beef747
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
April 21, 2025 07:41
beef747 to
6c70547
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
April 21, 2025 10:33
6c70547 to
b6530da
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
April 30, 2025 15:05
b6530da to
0f1deaa
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
May 27, 2025 09:44
0f1deaa to
2a62487
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
June 23, 2025 06:48
2a62487 to
94d41a1
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
July 10, 2025 13:43
94d41a1 to
b7c72ce
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
July 21, 2025 09:08
b7c72ce to
a35e4f1
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
August 26, 2025 07:53
a35e4f1 to
271acba
Compare
|
Your Vercel team Allen Zhou is not permitted to deploy from this git repository. Contact an administrator to add github organization descope as a Protected Git Scope in Allen Zhou on Vercel. Once added, commit again to see your changes. Learn more: https://vercel.com/docs/security/protected-git-scopes |
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
August 27, 2025 12:26
271acba to
5f35d0f
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
September 23, 2025 07:42
5f35d0f to
b043a09
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
2 times, most recently
from
October 23, 2025 19:50
06699bd to
88a12cc
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
November 25, 2025 09:37
88a12cc to
19c2700
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
December 24, 2025 08:15
19c2700 to
e70a212
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
January 29, 2026 10:07
e70a212 to
906dad0
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
March 5, 2026 11:02
906dad0 to
dd5d248
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
March 26, 2026 09:22
dd5d248 to
61f33b3
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
April 29, 2026 07:26
61f33b3 to
e627120
Compare
descope
Bot
force-pushed
the
renovate/major-7-material-ui-monorepo
branch
from
May 28, 2026 03:55
e627120 to
86e4fad
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
5.18.0→7.3.115.18.0→7.3.11Warning
Some dependencies could not be looked up. Check the Dependency Dashboard for more information.
Release Notes
mui/material-ui (@mui/icons-material)
v7.3.11Compare Source
A big thanks to the 5 contributors who made this release possible.
@mui/material@7.3.11startIconalignment (#48339) @mj12albertaria-checked=mixedwhen indeterminate (#48286) @mj12albertrolewithslotProps.input(#48472) @mj12albertDocs
Core
All contributors of this release in alphabetical order: @alexfauquette, @mj12albert, @oliviertassinari, @silviuaavram, @siriwatknp
v7.3.10Compare Source
A big thanks to the 15 contributors who made this release possible. A few highlights ✨:
@mui/material@7.3.10rootslot (#47916) @GerardasBnativeColoris used (#48144) @ZeeshanTamboliDocs
styledAPI docs for easier understanding (#47969) @olivier-lxoCore
All contributors of this release in alphabetical order: @Ahmad-Alaziz, @aman44444, @anchmelev, @arminmeh, @cherniavskii, @DanailH, @dav-is, @GerardasB, @mj12albert, @olivier-lxo, @pcorpet, @sai6855, @silviuaavram, @siriwatknp, @ZeeshanTamboli
v7.3.9Compare Source
A big thanks to the 15 contributors who made this release possible.
@mui/material@7.3.9color-mixvalue on top of default generated Material UI CSS variables (#47791) @ZeeshanTambolialphabefore color mixing to border bottom color when nativeColor + cssVariables is used (#47840) @ZeeshanTamboliDocs
(0, 1]behavior. (#47851) @matthias-ccrinativeColordocs (#47759) (#47789) @ZeeshanTamboliCore
All contributors of this release in alphabetical order: @alelthomas, @arthur-plazanet, @brijeshb42, @DanailH, @dav-is, @Janpot, @matthias-ccri, @mnajdova, @oliviertassinari, @pavan-sh, @PeterTYLiu, @sai6855, @silviuaavram, @vmakhaev, @ZeeshanTamboli
v7.3.8Compare Source
A big thanks to the 15 contributors who made this release possible. Here are some highlights ✨:
@mui/material@7.3.8disableCloseOnSelect(#47248) @ZeeshanTambolionDeletecheck (#47753) @ZeeshanTamboliaria-disabledfrom root span (#46318) @KirankumarAmbatipersistentandpermanentvariant Drawers should not override the styles via theme usingmodalclass (#47581) @ZeeshanTamboliReact.Keyinstead ofanyforkeyprop (#47619) @sonixx02Core
868d23e) @dav-isae29d03) @dav-iswindow.muiDocConfig(#47737) @JanpotDocs
All contributors of this release in alphabetical order: @aditya1906, @aemartos, @alelthomas, @bernardobelchior, @dav-is, @Janpot, @KirankumarAmbati, @mapache-salvaje, @nodirbekprogrammer, @Ocheretovich, @oliviertassinari, @sai6855, @silviuaavram, @sonixx02, @ZeeshanTamboli
v7.3.7Compare Source
A big thanks to the 16 contributors who made this release possible.
@mui/material@7.3.7squareprop on Accordion Root (#47457) @ZeeshanTamboliArrowLeft,Backspace&Deletebehavior for multiple and single-value rendering with proper caret handling (#47411) @jnbainclassesfrom root slot (#47519) @ZeeshanTambolicontained(#47499) @ZeeshanTamboliactiveElementinsideloopFocuson every keydown (#47566) @ZeeshanTamboliDocs
Core
@mui/internal-test-utilsto code infra repo (#47422) @JanpotAll contributors of this release in alphabetical order: @AarishMansur, @alelthomas, @atharva3333, @bricefrisco, @Janpot, @jnbain, @joserodolfofreitas, @mapache-salvaje, @mj12albert, @oliviertassinari, @prakhargupta1, @rithik56, @siriwatknp, @TimKraemer, @Zache, @ZeeshanTamboli
v7.3.6Compare Source
A big thanks to the 22 contributors who made this release possible.
@mui/material@7.3.6
filterSelectedOptionsdependency fromsyncHighlightedIndexuseCallback (#47378) @ZeeshanTambolisecondaryActionslot toListItem(#47399) @sai6855marksprop (#47370) @pcorpetownerStatespread intouseSnackbar(#47373) @ZeeshanTamboliapplyStyleswith custom color schemes (#47242) @akankshahu@mui/system@7.3.6
classattribute (#47200) @siriwatknp@mui/lab@7.3.6
Docs
elevationandsquareprops (#47261) @Ad1tya-007Core
next/fontfor local fonts (#47351) @Janpotappsfolder (#47183) @Janpotnext/font(#47347) @JanpotComponentLinkHeaderbottom margin (#47328) @Janpotnext.config.jstonext.config.mjs(#44040) @albarv340mui-materialpackage (#47392) @ZeeshanTamboliclean-csspackage from docs (#47314) @ZeeshanTambolibuidApiDocs->buildApiDocs(#47235) @Ad1tya-007@vitest/browserand@vitest/coverage-v8(#47189) @ZeeshanTamboliAll contributors of this release in alphabetical order: @Ad1tya-007, @akankshahu, @albarv340, @Aleksan4e3, @brijeshb42, @Copilot, @Fanzzzd, @Janpot, @JCQuintas, @kumarvishwajeettrivedi, @mapache-salvaje, @oliviertassinari, @pcorpet, @rifeplight, @rithik56, @rossdakin, @sai6855, @shamblonaut, @siriwatknp, @vrachuri28, @xBlizZer, @ZeeshanTamboli
v7.3.5Compare Source
A big thanks to the 11 contributors who made this release possible.
@mui/material@7.3.5theme.vars.shapemodule augmentation (#47164) @siriwatknp@mui/material-nextjs@7.3.5Docs
inputPropswithslotProps.inputin checkbox demos (#47083) @CODEKRUNCHCore
@mui/internal-test-utils(#47141) @bernardobelchiorbaseUrl(#47044) @Janpotlodashtoes-toolkit(#46974) @sukvvon|) inside code blocks (#47139) @ZeeshanTamboliAll contributors of this release in alphabetical order: @bernardobelchior, @brijeshb42, @CODEKRUNCH, @frncesc, @HeinrichFilter, @Janpot, @PossiblyAShrub, @sai6855, @siriwatknp, @sukvvon, @ZeeshanTamboli
v7.3.4Compare Source
A big thanks to the 3 contributors who made this release possible. Here are some highlights ✨:
Small update to revert a change that broke the
<Tabs>component. Also publishing a beta version of@mui/labwhich was accidentally published as a stable release.@mui/material@7.3.4Docs
Core
Configuration
📅 Schedule: (in timezone Asia/Jerusalem)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Never, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about these updates again.
This PR has been generated by Mend Renovate.