Skip to content

Update dependency kubernetes/kubernetes to v1.34.12 - #287

Open
gardener-ci-robot wants to merge 1 commit into
masterfrom
renovate/kubernetes-kubernetes-1.34.x
Open

gardener-ci-robot wants to merge 1 commit into
masterfrom
renovate/kubernetes-kubernetes-1.34.x

Conversation

@gardener-ci-robot

@gardener-ci-robot gardener-ci-robot commented Aug 20, 2026 •

Copy link
Copy Markdown
Collaborator

This PR contains the following updates:

Package Update Change
kubernetes/kubernetes patch v1.34.10 → v1.34.12

Release Notes

kubernetes/kubernetes (kubernetes/kubernetes)

v1.34.12

Compare Source

See kubernetes-announce@. Additional binary downloads are linked in the CHANGELOG.

See the CHANGELOG for more details.

v1.34.11

Compare Source

See kubernetes-announce@. Additional binary downloads are linked in the CHANGELOG.

See the CHANGELOG for more details.


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

Release note:

`kubernetes/kubernetes` has been updated to `v1.34.12`. [Release Notes](https://redirect.github.com/kubernetes/kubernetes/releases/tag/v1.34.12)

@gardener-ci-robot gardener-ci-robot added the kind/enhancement Enhancement, improvement, extension label Aug 20, 2026
@gardener-ci-robot
gardener-ci-robot requested a review from a team as a code owner August 20, 2026 19:26
@gardener-ci-robot gardener-ci-robot added the kind/enhancement Enhancement, improvement, extension label Aug 20, 2026
@gardener-prow

gardener-prow Bot commented Aug 20, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign holgerkoser for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@gardener-prow gardener-prow Bot added cla: yes Indicates the PR's author has signed the cla-assistant.io CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Aug 20, 2026
@gardener-ci-robot

Copy link
Copy Markdown
Collaborator Author

Edited/Blocked Notification

Renovate will not automatically rebase this PR, because it does not recognize the last commit author and assumes somebody else may have edited the PR.

You can manually request rebase by checking the rebase/retry box above.

⚠️ Warning: custom changes will be lost.

Signed-off-by: gardener-ci-robot <gardener.ci.robot@gmail.com>
@gardener-ci-robot
gardener-ci-robot force-pushed the renovate/kubernetes-kubernetes-1.34.x branch from 709ea2b to 2f4bb4b Compare September 23, 2026 20:37
@gardener-ci-robot gardener-ci-robot changed the title Update dependency kubernetes/kubernetes to v1.34.11 Update dependency kubernetes/kubernetes to v1.34.12 Sep 23, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

cla: yes Indicates the PR's author has signed the cla-assistant.io CLA. kind/enhancement Enhancement, improvement, extension size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant