Skip to content

Add rsync - #291

Open
anluerz wants to merge 1 commit into
gardener:masterfrom
anluerz:enh/add-rsync
Open

anluerz wants to merge 1 commit into
gardener:masterfrom
anluerz:enh/add-rsync

Conversation

@anluerz

@anluerz anluerz commented Sep 9, 2026

Copy link
Copy Markdown
Contributor

How to categorize this PR?
/area ops-productivity
/kind enhancement

What this PR does / why we need it:
Adds rsync to the Toolbelt. rsync is particularly useful for downloading files over flaky network links, as it can resume interrupted transfers rather than starting over.

Which issue(s) this PR fixes:
Fixes #

Special notes for your reviewer:

Release note:

[`rsync`](https://rsync.samba.org/) has been added.

Signed-off-by: Andreas Erz <andreas.erz@sap.com>
@anluerz
anluerz requested a review from a team as a code owner September 9, 2026 09:46
@gardener-prow gardener-prow Bot added area/ops-productivity Operator productivity related (how to improve operations) kind/enhancement Enhancement, improvement, extension size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Sep 9, 2026

@mimiteto mimiteto left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

Thanks for this PR. Sorry it took so much time but seems we matched perfectly our vacations.

@gardener-prow gardener-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 28, 2026
@gardener-prow

gardener-prow Bot commented Sep 28, 2026

Copy link
Copy Markdown

LGTM label has been added.

DetailsGit tree hash: 4e8cfc09cbedda765e5e02bc7ccb1f05e53919d9

@gardener-prow

gardener-prow Bot commented Sep 28, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: mimiteto

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@gardener-prow gardener-prow Bot added approved Indicates a PR has been approved by an approver from all required OWNERS files. needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. labels Sep 28, 2026
@gardener-prow

gardener-prow Bot commented Sep 28, 2026

Copy link
Copy Markdown

PR needs rebase.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/ops-productivity Operator productivity related (how to improve operations) kind/enhancement Enhancement, improvement, extension lgtm Indicates that a PR is ready to be merged. needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants