Skip to content

Add traceroute - #294

Merged
gardener-prow[bot] merged 1 commit into
gardener:masterfrom
anluerz:enh/add-traceroute
Sep 28, 2026
Merged

gardener-prow[bot] merged 1 commit into
gardener:masterfrom
anluerz:enh/add-traceroute

Conversation

@anluerz

@anluerz anluerz commented Sep 23, 2026

Copy link
Copy Markdown
Contributor

How to categorize this PR?
/area networking
/kind enhancement

What this PR does / why we need it:
Adds traceroute to the Toolbelt for diagnosing network routing and latency issues.

The package adds ~223KB to the installed image size (determined via docker run --rm ghcr.io/gardenlinux/gardenlinux:2150.9.0 apt-cache show traceroute | grep Installed-Size).

Which issue(s) this PR fixes:
Fixes #

Special notes for your reviewer:
/cc @gardener-core-networking-maintainers

Release note:

[`traceroute`](https://traceroute.sourceforge.net/) has been added.

Signed-off-by: Andreas Erz <andreas.erz@sap.com>
@anluerz
anluerz requested a review from a team as a code owner September 23, 2026 06:09
@gardener-prow gardener-prow Bot added area/networking Networking related kind/enhancement Enhancement, improvement, extension size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Sep 23, 2026
@axel7born

Copy link
Copy Markdown

/lgtm

@gardener-prow

gardener-prow Bot commented Sep 23, 2026

Copy link
Copy Markdown

@axel7born: adding LGTM is restricted to approvers and reviewers in OWNERS files.

Details

In response to this:

/lgtm

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@mimiteto mimiteto left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

Thanks for this PR. Sorry it took so much time but seems we matched perfectly our vacations.

@gardener-prow gardener-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 28, 2026
@gardener-prow

gardener-prow Bot commented Sep 28, 2026

Copy link
Copy Markdown

LGTM label has been added.

DetailsGit tree hash: d44938eb7080fdf048a8683f7f85a7d8333d937f

@gardener-prow

gardener-prow Bot commented Sep 28, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: mimiteto

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@gardener-prow gardener-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 28, 2026
@gardener-prow
gardener-prow Bot merged commit 7ded947 into gardener:master Sep 28, 2026
15 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/networking Networking related kind/enhancement Enhancement, improvement, extension lgtm Indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants