Skip to content

docs: add node conformance test instructions - #260

Open
Shelley-BaoYue wants to merge 1 commit into
kubeedge:masterfrom
Shelley-BaoYue:nodeconformance
Open

Shelley-BaoYue wants to merge 1 commit into
kubeedge:masterfrom
Shelley-BaoYue:nodeconformance

Conversation

@Shelley-BaoYue

@Shelley-BaoYue Shelley-BaoYue commented Sep 18, 2026 •

Copy link
Copy Markdown
Collaborator

What this PR does

  • Add node conformance test instructions

@kubeedge-bot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
To complete the pull request process, please assign shelley-baoyue after the PR has been reviewed.
You can assign the PR to them by writing /assign @shelley-baoyue in a comment when ready.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubeedge-bot kubeedge-bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Sep 18, 2026
@Shelley-BaoYue
Shelley-BaoYue force-pushed the nodeconformance branch 2 times, most recently from ab775bb to 6ced269 Compare September 18, 2026 07:40
Co-authored-by: Copilot App <223556219+Copilot@users.noreply.github.com>
Signed-off-by: Shelley-BaoYue <baoyue2@huawei.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants