Skip to content

build(deps-dev): bump sass from 1.104.0 to 1.104.1 - #1542

Merged
Rhilip merged 1 commit into
masterfrom
dependabot/npm_and_yarn/sass-1.104.1
Sep 19, 2026
Merged

Rhilip merged 1 commit into
masterfrom
dependabot/npm_and_yarn/sass-1.104.1

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 19, 2026

Copy link
Copy Markdown
Contributor

Bumps sass from 1.104.0 to 1.104.1.

Release notes

Sourced from sass's releases.

Dart Sass 1.104.1

To install Sass 1.104.1, download one of the packages below and add it to your PATH, or see the Sass website for full installation instructions.

Changes

  • Fix a bug where loud comments before @use rules could be emitted multiple times under certain circumstances.

  • Fix a bug in which an invalid selector at the end of an indented syntax stylesheet would cause the parser to crash instead of emitting a useful error message.

Command Line Interface

  • Many-to-many compilations no longer compile any files that appear in the output directory if the output directory is also within the source directory. This fixes a bug where --watch mode could enter an infinite loop recompiling the same CSS file over and over.

  • Sass now sets the modification time of output files to the time compilation started rather than the time it ended. This ensures that, if a source file is modified during compilation, --watch and --update mode will recompile the outputs to include the new source file contents.

See the full changelog for changes in earlier releases.

Changelog

Sourced from sass's changelog.

1.104.1

  • Fix a bug where loud comments before @use rules could be emitted multiple times under certain circumstances.

  • Fix a bug in which an invalid selector at the end of an indented syntax stylesheet would cause the parser to crash instead of emitting a useful error message.

Command Line Interface

  • Many-to-many compilations no longer compile any files that appear in the output directory if the output directory is also within the source directory. This fixes a bug where --watch mode could enter an infinite loop recompiling the same CSS file over and over.

  • Sass now sets the modification time of output files to the time compilation started rather than the time it ended. This ensures that, if a source file is modified during compilation, --watch and --update mode will recompile the outputs to include the new source file contents.

Commits
  • da84fb5 Fix --watch not recompiling sources modified during a compilation (#2850)
  • 8a287a7 Merge pull request #2859 from sass/fix-auto-merge
  • 6033b46 Add explicit versions to .github/util actions
  • cc71d43 Only auto-merge dependabot PRs once the tests pass
  • 49ddc4f Revert "Bump dart-lang/setup-dart in /.github/util/initialize"
  • 71a127d Avoid duplicating loud comments that appear before @use (#2854)
  • f19baab Merge pull request #2857 from sass/dependabot/github_actions/dot-github/util/...
  • f5263e2 Bump dart-lang/setup-dart in /.github/util/initialize
  • 3018c0d Merge pull request #2856 from sass/dependabot/github_actions/dart-lang/setup-...
  • 5d74adb Bump dart-lang/setup-dart from 1.8.0 to 1.8.1
  • Additional commits viewable in compare view

@dependabot dependabot Bot added dependencies Pull requests that update a dependency file javascript Pull requests that update javascript code labels Sep 19, 2026
Bumps [sass](https://github.com/sass/dart-sass) from 1.104.0 to 1.104.1.
- [Release notes](https://github.com/sass/dart-sass/releases)
- [Changelog](https://github.com/sass/dart-sass/blob/main/CHANGELOG.md)
- [Commits](sass/dart-sass@1.104.0...1.104.1)

---
updated-dependencies:
- dependency-name: sass
  dependency-version: 1.104.1
  dependency-type: direct:development
  update-type: version-update:semver-patch
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/sass-1.104.1 branch from 403d94b to 005e84d Compare September 19, 2026 14:10
@Rhilip
Rhilip merged commit f034683 into master Sep 19, 2026
2 checks passed
@Rhilip
Rhilip deleted the dependabot/npm_and_yarn/sass-1.104.1 branch September 19, 2026 14:11
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file javascript Pull requests that update javascript code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant